Skip to content

Feat/collaboration - #579

Open
diegosteiner wants to merge 3 commits into
masterfrom
feat/collaboration
Open

diegosteiner wants to merge 3 commits into
masterfrom
feat/collaboration

Conversation

@diegosteiner

@diegosteiner diegosteiner commented Oct 5, 2026 •

Copy link
Copy Markdown
Contributor

merge #576 first

@diegosteiner diegosteiner added this to the Oktober 2026 milestone Oct 5, 2026
@changeset-bot

changeset-bot Bot commented Oct 5, 2026 •

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 64360d3

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
Name Type
openscript-ch-website Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@diegosteiner diegosteiner mentioned this pull request Oct 6, 2026
@openscript
openscript added this pull request to stack #583 October 7, 2026 12:30
Comment thread res/sprite/binoculars.svg

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Hier sind noch ein paar Katzenschnautzhaare im SVG übrig geblieben.

### 1. Persönliches Kennenlernen

<Button href="mailto:hi@openscript.ch" withArrow>Kontakt aufnehmen</Button>
Wir lernen Sie am liebsten persönlich und bei Ihnen vor Ort kennen, um Sie und Ihre Bedürfnisse zu

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Wollen wir per Du mit den Besuchern sein? Auf Englisch sind wir immer per Du.


### 2. Analyse

Wir sichten Dokumentation, Quellcode, Infrastruktur, Abhängigkeiten, Datenflüsse und bestehende Deployments.

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Das passt gut, falls wir ein bestehendes Projekt übernehmen, für ein neues Projekt hingegen nicht so.

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Vielleicht können wir den Besucher mit einem Toggle zwischen Projekttypen wählen: neues Projekt, bestehendes Projekt und vom Weg abgekommenes Projekt.

dokumentieren die Ergebnisse.
</section>

<section class="white-background" style={{ paddingTop: "0" }} >

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Mir gefällt die Idee. Der übergang zum Baum ist aber etwas "hart".

This branch has not been deployed

No deployments
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants